treewide: remove more unmaintained gtk2 packages - #349744
Merged
Merged
Conversation
Aleksanaa
requested review from
Artturin,
astsmtl,
domenkozar,
flosse,
jb55,
prikhi,
sergei-mironov and
tckmn
October 19, 2024 11:25
Aleksanaa
force-pushed
the
treewide-drop-gtk2
branch
from
October 19, 2024 13:03
1c7a279 to
2b1d393
Compare
Artturin
reviewed
Oct 19, 2024
Aleksanaa
force-pushed
the
treewide-drop-gtk2
branch
2 times, most recently
from
October 22, 2024 03:21
a0e4705 to
61fdeaf
Compare
Aleksanaa
force-pushed
the
treewide-drop-gtk2
branch
from
October 25, 2024 14:43
61fdeaf to
f82db33
Compare
13 tasks
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Maintainers: if you are still using the package, you can request not to remove them.
Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Add a 👍 reaction to pull requests you find important.