Skip to content

feat(console-captor): Add ConsoleCaptor to collect console logs - #20

Merged
foxable merged 4 commits into
cronn:mainfrom
florian-cronn:feature/console-captor
Sep 7, 2026
Merged

feat(console-captor): Add ConsoleCaptor to collect console logs#20
foxable merged 4 commits into
cronn:mainfrom
florian-cronn:feature/console-captor

Conversation

@florian-cronn

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Sep 4, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2cf1b52

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@cronn/playwright-utils Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@foxable
foxable self-requested a review September 7, 2026 06:28
Comment thread docs/.vitepress/config.ts Outdated
Comment thread src/api/logging/console-captor.ts Outdated
Comment thread src/api/logging/console-captor.ts Outdated
Comment thread src/api/logging/console-captor.ts Outdated
Comment thread src/api/logging/console-captor.ts Outdated
Comment thread src/api/logging/console-captor.ts Outdated
Comment thread src/capturing/console-captor.ts
Comment thread src/api/logging/console-captor.ts Outdated
Comment thread src/api/logging/console-captor.ts Outdated
@foxable
foxable force-pushed the feature/console-captor branch from d40c93f to 2cf1b52 Compare September 7, 2026 08:56
@foxable
foxable merged commit d21484b into cronn:main Sep 7, 2026
2 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants