Conversation
🦋 Changeset detectedLatest commit: 7eae70b The changes in this PR will be included in the next version bump. This PR includes changesets to release 0 packagesWhen changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
mimarz
added this pull request to stack #5439
September 29, 2026 08:38
Contributor
|
Preview deployments for this pull request: storybook - |
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
September 29, 2026 08:44
ee97227 to
74773ca
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
September 29, 2026 12:06
74773ca to
d0907d2
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
2 times, most recently
from
September 29, 2026 13:33
f23d4b7 to
ee073ef
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
2 times, most recently
from
September 30, 2026 06:11
bc648fa to
642349d
Compare
eirikbacker
approved these changes
Sep 30, 2026
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
September 30, 2026 06:53
642349d to
b35a5a3
Compare
Barsnes
approved these changes
Sep 30, 2026
mimarz
force-pushed
the
config-schema-color-defaults
branch
2 times, most recently
from
September 30, 2026 07:44
9cc5e85 to
0936dd8
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
3 times, most recently
from
September 30, 2026 10:31
2d16ddf to
f339489
Compare
mimarz
removed this pull request from stack #5439
September 30, 2026 14:00
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
September 30, 2026 14:01
f339489 to
e095c8e
Compare
mimarz
changed the base branch from
config-output-tailwind
to
fix-tailwind-typo
September 30, 2026 14:01
mimarz
added this pull request to stack #5452
September 30, 2026 14:01
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
October 1, 2026 07:03
e095c8e to
816cdcc
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
October 1, 2026 08:10
816cdcc to
afe58b0
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
October 1, 2026 08:45
afe58b0 to
e2313e1
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
October 1, 2026 09:07
e2313e1 to
0b46135
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
October 1, 2026 09:35
0b46135 to
1663081
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
2 times, most recently
from
October 1, 2026 11:11
5b7d6a5 to
7f46a06
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
October 1, 2026 11:43
7f46a06 to
5b7e0f1
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
3 times, most recently
from
October 1, 2026 13:04
3a51084 to
8eb7c91
Compare
mimarz
force-pushed
the
config-schema-color-defaults
branch
from
October 1, 2026 13:22
8eb7c91 to
7eae70b
Compare
This branch has not been deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Better handling of how default severity colors where added when not defined by user.
Instead of having to remember everywhere we need to append the severity colors, its now updated to be part of the schema refinement.
This moves the handling to one central location so cleaner use elsewhere and simpler to change behaviour later.
Stack created with GitHub Stacks CLI • Give Feedback 💬