Conversation
mimarz
requested review from
Barsnes,
Febakke,
eirikbacker and
mrosvik
as code owners
September 30, 2026 14:01
🦋 Changeset detectedLatest commit: bab5d28 The changes in this PR will be included in the next version bump. This PR includes changesets to release 6 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
mimarz
added this pull request to stack #5452
September 30, 2026 14:01
Contributor
eirikbacker
approved these changes
Sep 30, 2026
mimarz
force-pushed
the
fix-tailwind-typo
branch
2 times, most recently
from
October 1, 2026 08:10
5fcbcd2 to
e8c0dbf
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
from
October 1, 2026 08:45
e8c0dbf to
7b8e4c4
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
from
October 1, 2026 09:07
7b8e4c4 to
e5d48aa
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
from
October 1, 2026 09:35
e5d48aa to
41c6736
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
from
October 1, 2026 09:54
41c6736 to
89dd2fe
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
2 times, most recently
from
October 1, 2026 11:43
17fe60d to
683fb23
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
from
October 1, 2026 12:37
683fb23 to
784cff1
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
2 times, most recently
from
October 1, 2026 13:04
784cff1 to
73d71df
Compare
mimarz
force-pushed
the
fix-tailwind-typo
branch
from
October 1, 2026 13:22
73d71df to
bab5d28
Compare
Barsnes
reviewed
Oct 2, 2026
| "@digdir/designsystemet-css": patch | ||
| --- | ||
|
|
||
| Fix `<theme>.tailwind.css` missing variables `--text-sm`, `--text-md` and `--text-lg`, mapped to the body font sizes. They were never generated because of a typo in the token name matching. Generate a new theme file to get the updated fixed file. |
Member
There was a problem hiding this comment.
Suggested change
| Fix `<theme>.tailwind.css` missing variables `--text-sm`, `--text-md` and `--text-lg`, mapped to the body font sizes. They were never generated because of a typo in the token name matching. Generate a new theme file to get the updated fixed file. | |
| Fix `<theme>.tailwind.css` missing variables `--text-sm`, `--text-md` and `--text-lg`, mapped to the body font sizes. Generate a new theme file to get the updated fixed file. |
Barsnes
approved these changes
Oct 2, 2026
This branch has not been deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Fix bug found in #5438. Making a separate PR so its easier to review fixed changes.