Skip to content

feat(pie-docs): DSW-3952 chip dropdown docs figma migration - #3121

Draft
talitha-ferreira-01 wants to merge 9 commits into
mainfrom
dsw-3952-chips-dropddown-figma-migration-docs
Draft

feat(pie-docs): DSW-3952 chip dropdown docs figma migration#3121
talitha-ferreira-01 wants to merge 9 commits into
mainfrom
dsw-3952-chips-dropddown-figma-migration-docs

Conversation

@talitha-ferreira-01

Copy link
Copy Markdown
Contributor

Describe your changes (can list changeset entries if preferable)

Author Checklist (complete before requesting a review)

  • I have performed a self-review of my code
  • I have added thorough tests where applicable (unit / component / visual)
  • I have reviewed the Docs site preview
  • If there are visual test updates, I have reviewed them properly before approving

Reviewer checklists (complete before approving)

Reviewer 1

  • I have reviewed the Docs site preview
  • If new links have been added, I have tested them in the preview
  • If there are visual test updates, I have reviewed them

Reviewer 2

  • I have reviewed the Docs site preview
  • If new links have been added, I have tested them in the preview
  • If there are visual test updates, I have reviewed them

@talitha-ferreira-01
talitha-ferreira-01 requested a review from a team as a code owner August 13, 2026 15:49
@changeset-bot

changeset-bot Bot commented Aug 13, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8f04541

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@justeattakeaway/pie-docs Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pie-design-system-app

Copy link
Copy Markdown
Contributor



Fails
🚫 You have unchecked checklist items outside the "Not-applicable Checklist items" section.

Please ensure all unchecked checkboxes are moved to the appropriate section.

🚫 You have unchecked checklist items in Reviewer 1's section.

Please ensure all items are addressed before approval.

🚫 You have unchecked checklist items in Reviewer 2's section.

Please ensure all items are addressed before approval.

Generated by 🚫 dangerJS against 8f04541

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant