Skip to content

Dsw 3906 textarea uncontrolled approach2 - #3137

Draft
KatarinaNeskovic wants to merge 7 commits into
mainfrom
dsw-3906-textarea-uncontrolled-approach2
Draft

Dsw 3906 textarea uncontrolled approach2#3137
KatarinaNeskovic wants to merge 7 commits into
mainfrom
dsw-3906-textarea-uncontrolled-approach2

Conversation

@KatarinaNeskovic

Copy link
Copy Markdown
Contributor

Please click the Preview tab and select a different PR template if your change isn't for an existing web component.

PIE Docs Template
New Web Component Template

═══════════════════════════════════════════════════════════

Describe your changes (can list changeset entries if preferable)

Author Checklist (complete before requesting a review, do not delete any)

  • I have performed a self-review of my code.
  • I have added thorough tests where applicable (unit / component / visual).
  • I have reviewed the PIE Storybook/PIE Docs PR preview.
  • I have reviewed visual test updates properly before approving.
  • If changes will affect consumers of the package, I have created a changeset entry.
  • If a changeset file has been created, I have tested these changes in PIE Aperture using the /test-aperture command.

Not-applicable Checklist items

Please move any Author checklist items that do not apply to this pull request here.


Reviewer checklists (complete before approving)

Mark items as [-] N/A if not applicable.

Reviewer 1

  • I have reviewed the PIE Storybook/PIE Docs PR preview.
  • I have verified that all acceptance criteria for this ticket have been completed.
  • I have reviewed the Aperture changes (if added)
  • If there are visual test updates, I have reviewed them.

Reviewer 2

  • I have reviewed the PIE Storybook/PIE Docs PR preview.
  • I have verified that all acceptance criteria for this ticket have been completed.
  • I have reviewed the Aperture changes (if added)
  • If there are visual test updates, I have reviewed them.

@changeset-bot

changeset-bot Bot commented Aug 19, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 96230ad

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
@justeattakeaway/pie-textarea Minor
@justeattakeaway/pie-storybook Minor
@justeattakeaway/pie-webc Patch
@justeattakeaway/pie-docs Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant