Skip to content

feat(pie-docs): DSW-4280 tooltip placement and offset update - #3150

Open
tingting-lau wants to merge 4 commits into
mainfrom
dsw-4280-tooltip-placement-and-offset-update
Open

feat(pie-docs): DSW-4280 tooltip placement and offset update#3150
tingting-lau wants to merge 4 commits into
mainfrom
dsw-4280-tooltip-placement-and-offset-update

Conversation

@tingting-lau

@tingting-lau tingting-lau commented Aug 24, 2026

Copy link
Copy Markdown
Contributor

Describe your changes (can list changeset entries if preferable)

Author Checklist (complete before requesting a review)

  • I have performed a self-review of my code
  • I have added thorough tests where applicable (unit / component / visual)
  • I have reviewed the Docs site preview
  • If there are visual test updates, I have reviewed them properly before approving

Reviewer checklists (complete before approving)

Reviewer 1 @raoufswe

  • I have reviewed the Docs site preview
  • If new links have been added, I have tested them in the preview
  • If there are visual test updates, I have reviewed them

Reviewer 2 @talitha-ferreira-01

  • I have reviewed the Docs site preview
  • If new links have been added, I have tested them in the preview
  • If there are visual test updates, I have reviewed them

@tingting-lau
tingting-lau requested a review from a team as a code owner August 24, 2026 09:09
@changeset-bot

changeset-bot Bot commented Aug 24, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f7c5753

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@justeattakeaway/pie-docs Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

raoufswe
raoufswe previously approved these changes Aug 24, 2026
@talitha-ferreira-01 talitha-ferreira-01 changed the title DSW-4280 tooltip placement and offset update.md feat(pie-docs): DSW-4280 tooltip placement and offset update Aug 24, 2026
@raoufswe
raoufswe self-requested a review August 24, 2026 14:28
@pie-design-system-app

Copy link
Copy Markdown
Contributor

Fails
🚫 You have unchecked checklist items outside the "Not-applicable Checklist items" section.

Please ensure all unchecked checkboxes are moved to the appropriate section.

Generated by 🚫 dangerJS against f7c5753

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants