Skip to content

[ko] Update content/ko/docs/setup/production-environment/tools/kubead… - #57636

Open
yucori wants to merge 1 commit into
kubernetes:mainfrom
yucori:troubleshooting-kubeadm-ko
Open

yucori wants to merge 1 commit into
kubernetes:mainfrom
yucori:troubleshooting-kubeadm-ko

Conversation

@yucori

@yucori yucori commented Sep 20, 2026

Copy link
Copy Markdown
Contributor

Description

Update Korean Translation of content/en/docs/setup/production-environment/tools/kubeadm/troubleshooting-kubeadm.md

Issue

Closes: #57635

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 20, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign seokho-son for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added area/localization General issues or PRs related to localization language/ko Issues or PRs related to Korean language sig/docs Categorizes an issue or PR as relevant to SIG Docs. labels Sep 20, 2026
@netlify

netlify Bot commented Sep 20, 2026

Copy link
Copy Markdown

Pull request preview available for checking

Built without sensitive environment variables

Name Link
🔨 Latest commit f49cc14
🔍 Latest deploy log https://app.netlify.com/projects/kubernetes-io-main-staging/deploys/6aaf9af3bf99630008876b61
😎 Deploy Preview https://deploy-preview-57636--kubernetes-io-main-staging.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area/localization General issues or PRs related to localization cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. language/ko Issues or PRs related to Korean language sig/docs Categorizes an issue or PR as relevant to SIG Docs. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

[ko] Update content/ko/docs/setup/production-environment/tools/kubeadm/troubleshooting-kubeadm/

1 participant